Skip to content

Update CODEOWNERS to AllenCell team - #117

Merged
jessicasyu merged 1 commit into
mainfrom
admin/update-codeowners
Sep 21, 2026
Merged

jessicasyu merged 1 commit into
mainfrom
admin/update-codeowners

Conversation

@jessicasyu

Copy link
Copy Markdown

Problem

The current CODEOWNERS points to a team that existed in previous organization. When this repo was migrated to AllenCell, it no longer had access to this team, and therefore the CODEOWNERS file could not assign reviewers.

Solution

Added a new team AllenCell/simviz-web-devs with membership matching the original team, and updated CODEOWNERS to point to this team.

Type of change

  • Bug fix (non-breaking change which fixes an issue)

@toloudis

Copy link
Copy Markdown
Contributor

it should be noted that this is a Python package and not really something that web devs would review, at least if you go by the team name

@jessicasyu

Copy link
Copy Markdown
Author

@toloudis I don't feel strongly about the team name, happy to change or add a separate simviz-python-devs team?

@ShrimpCryptid

Copy link
Copy Markdown
Contributor

@jessicasyu I think this is fine! We can keep it as web-devs for now, and change it to Python if we have another SimViz Python repo that's being actively developed

@jessicasyu
jessicasyu merged commit 42ee493 into main Sep 21, 2026
1 check passed
@jessicasyu
jessicasyu deleted the admin/update-codeowners branch September 21, 2026 15:13
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants