Bump mozilla/addons/.github/actions/slack from 43401a931ebc09f2e511deed766171b0105414bc to 46043277ef3a3a0bdfe0e27414c29932a25d4560 in /.github/actions/slack-workflow-notification - #16460
Merged
Merged
Conversation
Bumps [mozilla/addons/.github/actions/slack](https://github.com/mozilla/addons) from 43401a9 to 4604327. - [Commits](43401a9...4604327) --- updated-dependencies: - dependency-name: mozilla/addons/.github/actions/slack dependency-version: 4604327 dependency-type: direct:production ... Signed-off-by: dependabot[bot] <support@github.com>
diox
approved these changes
Sep 23, 2026
diox
deleted the
dependabot/github_actions/dot-github/actions/slack-workflow-notification/mozilla/addons/dot-github/actions/slack-46043277ef3a3a0bdfe0e27414c29932a25d4560
branch
September 23, 2026 09:39
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Bumps mozilla/addons/.github/actions/slack from 43401a9 to 4604327.
Commits
4604327Bump docker/login-action in /.github/actions/login-gar (#16434)6e5c4f7Bump docker/login-action in /.github/actions/login-docker (#16432)6abe4caBump slackapi/slack-github-action from 2.1.1 to 4.0.0 in /.github/actions/sla...c6d2e38action: update the comment left by the release automation (#16453)c1a2defRemove obsolete scripts/transfer-issues.sh script (#16450)69e2106Bump actions/deploy-pages from 5.0.0 to 5.0.1 (#16442)442a703action: release: add a comment to the PR with the link to the draft release (...00817a7Add CODEOWNERS for automatic review assignment (#16449)3ebcbd4Create a new 'release' GitHub action (#16446)6bbc830Update dependabot.yml (#16431)Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)